Skip to content

[Fix] Valkyr teeth - #381

Open
bzzzbzzbz wants to merge 1 commit into
ArcaneSS14:masterfrom
bzzzbzzbz:valkyr_fix
Open

bzzzbzzbz wants to merge 1 commit into
ArcaneSS14:masterfrom
bzzzbzzbz:valkyr_fix

Conversation

@bzzzbzzbz

@bzzzbzzbz bzzzbzzbz commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

Описание PR

Починил зубы валькирии.

Тип PR

  • Feature
  • Fix
  • Tweak
  • Balance
  • Refactor
  • Port
  • Translate
  • Resprite

Лицензионное соглашение

  • Я согласен с лицензионным соглашением Contributor License Agreement, для Arcane SS14 Organization.

Изменения

🆑 seemah

  • fix: Исправлены зубы головы валькирии.

@coderabbitai

coderabbitai Bot commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

Important

Review skipped

Review was skipped due to path filters

⛔ Files ignored due to path filters (1)
  • Resources/Textures/_Arcane/Mobs/Customization/HumanoidXeno/head.rsi/valkyr_teeth.png is excluded by !**/*.png

CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including **/dist/** will override the default block on the dist directory, by removing the pattern from both the lists.

⚙️ Run configuration

Configuration used: Path: .coderabbit.yaml

Review profile: ASSERTIVE

Plan: Advanced

Run ID: 1ed8b6f1-aac4-4b39-b022-1fb7b69bbf0b

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out.

❤️ Share

Comment @coderabbitai help to get the list of available commands.

@github-actions

Copy link
Copy Markdown
Contributor

RSI Diff Bot; head commit f93b98f merging into b5aa71a
This PR makes changes to 1 or more RSIs. Here is a summary of all changes:

Resources/Textures/_Arcane/Mobs/Customization/HumanoidXeno/head.rsi

State Old New Status
valkyr_teeth Modified

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant