Components & Modularization - #7
martindale wants to merge 20 commits into
Conversation
Codecov Report❌ Patch coverage is
Additional details and impacted files@@ Coverage Diff @@
## master #7 +/- ##
===========================================
+ Coverage 43.45% 62.56% +19.11%
===========================================
Files 6 6
Lines 741 764 +23
===========================================
+ Hits 322 478 +156
+ Misses 419 286 -133 ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
|
Important Review skippedToo many files! This PR contains 1474 files, which is 1374 over the limit of 100. To get a review, reduce the PR to 100 files or fewer by splitting it into smaller PRs or changing its base branch. Upgrade to a paid plan to raise the limit. Usage-priced reviews support at most 300 files. ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Pro Plus Run ID: ⛔ Files ignored due to path filters (2055)
📒 Files selected for processing (1474)
You can disable this status message by setting the |
General cleanup
Goals
browser.jscontentNext Steps
scripts/browser.jstotypes/client.jsor similar