Skip to content

Add science module leaders to code ownership - #146

Open
Maggie (maggiehendry) wants to merge 13 commits into
MetOffice:mainfrom
maggiehendry:86-enable-github-code-ownership-for-jules
Open

Maggie (maggiehendry) wants to merge 13 commits into
MetOffice:mainfrom
maggiehendry:86-enable-github-code-ownership-for-jules

Conversation

@maggiehendry

@maggiehendry Maggie (maggiehendry) commented Aug 24, 2026

Copy link
Copy Markdown
Collaborator

A request was sent by email to the module leaders on Thursday, August 06, 2026 with information regarding the automatic review process the GitHub CODEOWNER file enables and to request a list of directories/files that required approvals for each module along with GitHub usernames. This information has been added to the CODEOWNER file. The supplier of the information has been added to the following table. Not all module leaders responded so where this was the case I have made an informed guess. This is also indicated on the table. Several module leaders are not on the https://github.com/orgs/MetOffice/teams/simsyssciencereviewers list. In order to complete a review module leaders need to request access to this list by contacting @MetOffice/ssdteam.

This question has recently come up while trying to identify JULES sci/tech reviewers. What is the minimum set up that an owner needs to complete in order to approve changes and carry out Sci/tech reviews if they are not themselves involved in active development. It would be useful to have this information available.

Science module Module leader(s) List supplied by
Surface John Edwards (John Edwards (@JMEdwardsXtr)) John Edwards (@JMEdwardsXtr)
Hydrology Nic Gedney (Nic Gedney (@nicgedney)) and Sonja Folwell (Sonja (@sfolwell)) Maggie (@maggiehendry)
Vegetation Lina Mercado and Rebecca Oliver (Becky Oliver (@beckyoliver)) Maggie (@maggiehendry)
Veg3 RED Demography Andy Wiltshire (andy-wiltshire) and Arthur Argles (Arthur Argles (@mo-arthurargles)) Arthur Argles (@mo-arthurargles)
Biogeochemistry Eleanor Burke (Eleanor Burke (@eleanorgb)) and Noah Smith (Noah Smith (@noahdsmith)) Eleanor Burke (@eleanorgb)
Biogenic fluxes Garry Hayman (GarryHayman) and James Weber Maggie (@maggiehendry)
Fire Doug Kelley and Chantelle Burton (chantelleburton) Maggie (@maggiehendry)
Lakes Helen Johnson (Helen Johnson (@HelenRJohnson)) Helen Johnson (@HelenRJohnson)
CABLE Science Ian Harman and Claire Carouge (Claire Carouge (@ccarouge)) Maggie (@maggiehendry)
IMOGEN Chris Huntingford and Eleanor Burke (Eleanor Burke (@eleanorgb)) Eleanor Burke (@eleanorgb)
Evaluation Heather Rumbold and Siyuan Tian Not associated with changes in this repository

PR Summary

Science module leaders and the code they have responsibility for have been added to the GitHub code owners file.

Sci/Tech Reviewer:
Code Reviewer: Sam Clarke-Green (@t00sa)

Code Quality Checklist

(Some checks are automatically carried out via the CI pipeline)

  • I have performed a self-review of my own code
  • My code follows the project's style guidelines
  • Comments have been included that aid understanding and enhance the readability of the code
  • My changes generate no new warnings
  • If editing rose-meta/jules-shared then have you supplied a linked UM and LFRic Apps PR?

Testing

  • I have tested this change locally, using the JULES rose-stem suite
  • If shared files have been modified, I have run the UM and LFRic Apps rose stem suites
  • If any tests fail (rose-stem or CI) the reason is understood and acceptable (eg. kgo changes)
  • I have added tests to cover new functionality as appropriate (eg. system tests, unit tests, etc.)

Changes to CODEOWNER file cannot be tested until the PR is merged onto main.

trac.log

Security Considerations

  • I have reviewed my changes for potential security issues
  • Sensitive data is properly handled (if applicable)
  • Authentication and authorisation are properly implemented (if applicable)

Performance Impact

  • Performance of the code has been considered and, if applicable, suitable performance measurements have been conducted

AI Assistance and Attribution

  • Some of the content of this change has been produced with the assistance of Generative AI tool name (e.g., Met Office Github Copilot Enterprise, Github Copilot Personal, ChatGPT GPT-4, etc) and I have followed the Simulation Systems AI policy (including attribution labels)

Documentation

  • Where appropriate I have updated documentation related to this change and confirmed that it builds correctly

Approvals

Please request all relevant approvals. See the CodeOwners.txt file for section owners.

Technical

  • JULES Code Owner
  • OpenMP
  • River Routing
  • Rose Stem
  • Rose Metadata
  • Upgrade Macros

Scientific

  • Surface
  • Hydrology
  • Vegetation
  • Veg3 RED Demography
  • Biogeochemistry
  • Biogenic fluxes
  • Fire
  • Lakes
  • Evaluation
  • Imogen

Sci/Tech Review

  • I understand this area of code and the changes being added
  • The proposed changes correspond to the pull request description
  • Documentation is sufficient (do documentation papers need updating)
  • Sufficient testing has been completed

Please alert the code reviewer via a tag when you have approved the SR

Code Review

  • All dependencies have been resolved
  • Related Issues have been properly linked and addressed
  • CLA compliance has been confirmed
  • Code quality standards have been met
  • Tests are adequate and have passed
  • Documentation is complete and accurate
  • Security considerations have been addressed
  • Performance impact is acceptable

@github-actions github-actions Bot added the cla-required The CLA has not yet been signed by the author of this PR - added by GA label Aug 24, 2026
@github-actions github-actions Bot added cla-modified The CLA has been modified as part of this PR - added by GA and removed cla-required The CLA has not yet been signed by the author of this PR - added by GA labels Sep 8, 2026
@github-actions github-actions Bot removed the cla-modified The CLA has been modified as part of this PR - added by GA label Sep 9, 2026
@maggiehendry

Copy link
Copy Markdown
Collaborator Author

An email has been sent to the science module leaders asking for approval.

Comment thread .github/CODEOWNERS
Comment thread .github/CODEOWNERS

@nicgedney Nic Gedney (nicgedney) left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

My details are correct. Thanks.

@noahdsmith Noah Smith (noahdsmith) left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

My details are correct, thanks!

@maggiehendry

Copy link
Copy Markdown
Collaborator Author

My details are correct, thanks!

Thanks Noah Smith (@noahdsmith). I've added your username to the codeowner file as I hadn't included this before. Are you able to approve the change in the reviewers panel please? Thanks.

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Approved. We may need to update wider documentation if a future change adds a new file under a module area in CODEWONERS. Will/has this be done done for vn8.3?

Comment thread .github/CODEOWNERS

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks. My details are correct. There is currently a vacancy in the vegetation module leaders, so when this is filled can this be updated?

@GarryHayman GarryHayman left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

My details are correct. Thanks Garry

@chantelleburton chantelleburton left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

My details are correct. Thanks.

@maggiehendry

Copy link
Copy Markdown
Collaborator Author

Approved. We may need to update wider documentation if a future change adds a new file under a module area in CODEWONERS. Will/has this be done done for vn8.3?

Thanks Arthur Argles (@mo-arthurargles). The file can be updated as required. If the code was structured differently there would be less additions in the CODEOWNER file for example, if there was a separate directory containing "red" related code then this could be specified on one line.

**/red @mo-arthurargles

Then any file in a "red" directory in the source tree would require your approval. CABLE for example:

**/cable			@ccarouge

From the one instruction, the CABLE code owner will receive a review request for changes in:

./src/control/cable
./src/initialisation/cable

Something to think about when organising code.

@maggiehendry

Copy link
Copy Markdown
Collaborator Author

Thanks. My details are correct. There is currently a vacancy in the vegetation module leaders, so when this is filled can this be updated?

Thanks Becky Oliver (@beckyoliver). The file can be updated when required. 😄

@maggiehendry

Copy link
Copy Markdown
Collaborator Author

Thanks for all your approvals so far. Just so that you are aware I've upgraded to main and added the metadata directories that were committed. 13cad69

@maggiehendry

Copy link
Copy Markdown
Collaborator Author

Sam Clarke-Green (@t00sa) should I pass this directly to code review after the module leaders have all approved?

@eleanorgb

Copy link
Copy Markdown
Contributor

I approve - no idea how to tick though?

@maggiehendry

Maggie (maggiehendry) commented Sep 14, 2026

Copy link
Copy Markdown
Collaborator Author

I approve - no idea how to tick though?

Eleanor Burke (@eleanorgb) thanks! To approve, go to the "Files changed" tab, click the "Submit review", select the "Approve" radio button with or without a comment then "Submit review" 😄

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Enable github code ownership for Jules

10 participants