Response to Ned's PR - #29
Open
augustine-procogia wants to merge 4 commits into
Open
Conversation
…cussed during the code review. I also updated the assignment operator used throughout the package
|
@augustine-procogia Can you help clean out the code so that it will only reflect the changes that you made? It will be hard to review this with all the code marked as changed. Currently this is showing 57 files are changed. and it will be hard to do a code review for the specific changes. |
Author
|
@rvitancol |
added 3 commits
March 22, 2022 17:33
… the nanopipeline preDE and DE modules
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
I have updated the package to reflect the outcome of the code review and implemented the required changes. The package builds with the following command
devtools::build(vignettes = F).