Skip to content

feat(RHINENG-29493): add Satellite option to Workload fitler - #1754

Merged
katarinazaprazna merged 5 commits into
RedHatInsights:masterfrom
computercamplove:RHINENG-29493
Aug 13, 2026
Merged

katarinazaprazna merged 5 commits into
RedHatInsights:masterfrom
computercamplove:RHINENG-29493

Conversation

@computercamplove

Copy link
Copy Markdown
Contributor

Description

Associated Jira ticket: https://redhat.atlassian.net/browse/RHINENG-29493

Added Satellite workload option to the local Systems table workload filter. The new option sends filter[system_profile][satellite]=true to the API when selected.

How to test the PR

before testing make sure backend change is merged RedHatInsights/patchman-engine#2301

  1. visit Content systems page
  2. Select Workload filter - select Satellite option

Before the change

  • Workload filter menu without "Satellite" option

After the change

  • Workload filter menu with "Satellite" option
  • Filtering by "Satellite" option works as expected

Dependent work link

RedHatInsights/patchman-engine#2301

Checklist:

  • The commit message has the Jira ticket linked
  • PR has a short description
  • Screenshots before and after the change are added
  • Tests for the changes have been added
  • README.md is updated if necessary
  • Needs additional dependent work

@computercamplove
computercamplove requested a review from a team as a code owner August 12, 2026 10:13
@computercamplove computercamplove self-assigned this Aug 12, 2026
@katarinazaprazna katarinazaprazna self-assigned this Aug 12, 2026

@katarinazaprazna katarinazaprazna left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the PR! Just fixing the unit tests real quick

@codecov-commenter

codecov-commenter commented Aug 13, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 77.60%. Comparing base (2c65dcf) to head (7d5c565).

Additional details and impacted files
@@           Coverage Diff           @@
##           master    #1754   +/-   ##
=======================================
  Coverage   77.60%   77.60%           
=======================================
  Files         102      102           
  Lines        3277     3278    +1     
  Branches      753      751    -2     
=======================================
+ Hits         2543     2544    +1     
  Misses        655      655           
  Partials       79       79           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@katarinazaprazna
katarinazaprazna enabled auto-merge (rebase) August 13, 2026 17:12
@katarinazaprazna

Copy link
Copy Markdown
Collaborator

/retest

2 similar comments
@katarinazaprazna

Copy link
Copy Markdown
Collaborator

/retest

@katarinazaprazna

Copy link
Copy Markdown
Collaborator

/retest

@katarinazaprazna
katarinazaprazna merged commit 08f1cea into RedHatInsights:master Aug 13, 2026
5 of 6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants