Skip to content

LWLP-1233: track sent advisory created messages - #1724

Merged
jlsherrill merged 1 commit into
content-services:mainfrom
jlsherrill:1233
Sep 15, 2026
Merged

jlsherrill merged 1 commit into
content-services:mainfrom
jlsherrill:1233

Conversation

@jlsherrill

Copy link
Copy Markdown
Member

Summary

Previously when i added the sending of advisory-created messages, i forgot to check if it was resent. Here i'm reusing the existing ListUnnotifiedAdvisories() call and MarkAsNotified() and re-using the orgId field to track if its used.

Testing steps

@jlsherrill
jlsherrill requested a review from a team as a code owner September 11, 2026 01:54
@jlsherrill

jlsherrill commented Sep 11, 2026

Copy link
Copy Markdown
Member Author

note, i'm abusing the org id column here a little bit, we could rename this column to something like 'DestinationId' or something, but i'd probably prefer to do that as a followup, and it would be a bit complicated to do in a reversible/deployable way

xbhouse
xbhouse previously approved these changes Sep 14, 2026
Comment thread pkg/external_repos/commands/sync_lightwell_advisories.go Outdated
@jlsherrill

Copy link
Copy Markdown
Member Author

/retest

@jlsherrill
jlsherrill merged commit b4f0c96 into content-services:main Sep 15, 2026
17 of 20 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants