Skip to content

DEVO-338 - Add DevOps teams to CODEOWNERS - #16

Closed
im-org-management-bot wants to merge 1 commit into
mainfrom
DEVO-338-add-teams-to-codeowners
Closed

im-org-management-bot wants to merge 1 commit into
mainfrom
DEVO-338-add-teams-to-codeowners

Conversation

@im-org-management-bot

Copy link
Copy Markdown
Contributor

For https://im-jira.internal.towerswatson.com/browse/DEVO-338 we are adding DevOps teams to CODEOWNERS file

@im-org-management-bot
im-org-management-bot requested a review from a team as a code owner February 6, 2026 16:47
@ahernandez411
ahernandez411 deleted the DEVO-338-add-teams-to-codeowners branch February 6, 2026 16:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants