Skip to content

PLPT-1149 - Add IMBA Squad to CODEOWNERS - #35

Merged
ahernandez411 merged 1 commit into
mainfrom
im-open--update-im-open-codeowners
Aug 26, 2025
Merged

ahernandez411 merged 1 commit into
mainfrom
im-open--update-im-open-codeowners

Conversation

@im-org-management-bot

Copy link
Copy Markdown
Contributor

Changes made for: PLPT-1149

The following replacement rules were applied:

  • Add IMBA Squad to CODEOWNERS

@im-org-management-bot
im-org-management-bot requested a review from a team August 26, 2025 20:19
@ahernandez411
ahernandez411 merged commit b2f9b4b into main Aug 26, 2025
3 checks passed
@ahernandez411
ahernandez411 deleted the im-open--update-im-open-codeowners branch August 26, 2025 20:37
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants