Skip to content

PLPT-1149 - Add IMBA Squad to CODEOWNERS - #17

Merged
ahernandez411 merged 1 commit into
mainfrom
im-open--update-im-open-codeowners
Aug 26, 2025
Merged

ahernandez411 merged 1 commit into
mainfrom
im-open--update-im-open-codeowners

Conversation

@im-org-management-bot

Copy link
Copy Markdown
Contributor

Changes made for: PLPT-1149

The following replacement rules were applied:

  • Add IMBA Squad to CODEOWNERS

@im-org-management-bot
im-org-management-bot requested a review from a team August 26, 2025 20:26
@github-actions

Copy link
Copy Markdown
File % Stmts % Branch % Funcs % Lines Uncovered Line #s
All files 61.97 51.17 72 61.91
src 71.11 54.05 88.89 70.9
variableSubstitution.ts 71.11 54.05 88.89 70.9 ...,191-192,207-216
src/operations 58.91 50.37 68.29 58.9
envVariableUtility.ts 100 84 100 100 15,45-67
fileEncodingUtility.ts 68.75 50 100 67.74 ...3,29,44-50,60,66
jsonVariableSubstitutionUtility.ts 100 76.47 100 100 20-46
utility.ts 9.43 6 13.33 9.55 ...0,71-383,391-394
xmlDomUtility.ts 92.16 69.44 100 91.84 27,66,69,90
xmlVariableSubstitution.ts 91.21 86.76 100 91.21 ...39,69-70,119-120

=============================== Coverage summary ===============================

Statements : 61.97% ( 334/539 )

Branches : 51.17% ( 175/342 )

Functions : 72% ( 36/50 )

Lines : 61.91% ( 330/533 )

================================================================================

@ahernandez411
ahernandez411 merged commit 730b0d3 into main Aug 26, 2025
3 checks passed
@ahernandez411
ahernandez411 deleted the im-open--update-im-open-codeowners branch August 26, 2025 20:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants