Skip to content

[deepin-font-manager] Updates for project Deepin Font Manager - #213

Merged
lzwind merged 2 commits into
masterfrom
transifex_update_950be2762b
Sep 1, 2026
Merged

lzwind merged 2 commits into
masterfrom
transifex_update_950be2762b

Conversation

@transifex-integration

@transifex-integration transifex-integration Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

The following localization files have been updated:

Details

Parameter Value
Transifex Project Deepin Font Manager
Grouping All Resources and Languages

Files

Source File Translation File Resource Language Type Mode
translations/deepin-font-manager.ts translations/deepin-font-manager_sv.ts translations/deepin-font-manager.ts (master) sv TRANSLATION default
translations/desktop/desktop.ts translations/desktop/desktop_sv.ts translations/desktop/desktop.ts (master) sv TRANSLATION default

100% translated source file: 'deepin-font-manager.ts'
on 'sv'.
100% translated source file: 'desktop.ts'
on 'sv'.
@deepin-ci-robot

Copy link
Copy Markdown

Hi @transifex-integration[bot]. Thanks for your PR.

I'm waiting for a linuxdeepin member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@deepin-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: lzwind, transifex-integration[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@lzwind
lzwind merged commit 40e90c9 into master Sep 1, 2026
22 of 24 checks passed
@transifex-integration
transifex-integration Bot deleted the transifex_update_950be2762b branch September 19, 2026 08:52
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants