Skip to content

fix: refine TTY switching - #534

Merged
zccrs merged 1 commit into
linuxdeepin:masterfrom
calsys456:master
Aug 28, 2025
Merged

zccrs merged 1 commit into
linuxdeepin:masterfrom
calsys456:master

Conversation

@calsys456

Copy link
Copy Markdown
Contributor
  1. Add callback for disableRender, to ensure render is exactly disabled before DDM doing further actions;
  2. Revoke all evdev devices before VT switch (at disableRender()), to prevent accidental input events during switch;
  3. Refine logic of switchToUser, take care of showing lockscreen after switch;
  4. Make event function args const

PR for ddm: linuxdeepin/ddm#39
PR for treeland-protocol: linuxdeepin/treeland-protocols#23

@sourcery-ai sourcery-ai Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sorry @apr3vau, you have reached your weekly rate limit for Sourcery. Please try again later

@deepin-ci-robot

Copy link
Copy Markdown

Hi @apr3vau. Thanks for your PR.

I'm waiting for a linuxdeepin member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Comment thread src/modules/ddm/ddminterfacev1.cpp Outdated
1. Add callback for disableRender, to ensure render is exactly disabled
before DDM doing further actions;
2. Revoke all evdev devices before VT switch (at disableRender()), to
prevent accidental input events during switch;
3. Refine logic of switchToUser, take care of showing lockscreen after
switch;
4. Make event function args const
@deepin-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: apr3vau, zccrs

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@zccrs
zccrs merged commit 18026e2 into linuxdeepin:master Aug 28, 2025
9 of 12 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants