Skip to content

(Auto) Sync main-to-development - #2272

Closed
github-actions[bot] wants to merge 13 commits into
developmentfrom
main
Closed

github-actions[bot] wants to merge 13 commits into
developmentfrom
main

Conversation

@github-actions

Copy link
Copy Markdown

This is an automatically generated PR to sync main to development.

FiniteStateGit and others added 7 commits February 23, 2023 13:53
Signed-off-by: Jonathan Capes <j.capes@gmail.com>
…nsole-verification

Add google search console verification
Signed-off-by: chanmosq <75444793+chanmosq@users.noreply.github.com>
Signed-off-by: GPC-94301 <99716227+LB-JaroslawGaweda@users.noreply.github.com>
Signed-off-by: GPC-94301 <99716227+LB-JaroslawGaweda@users.noreply.github.com>
* Revert "Added information about O3DE source build (#2233)"

This reverts commit 14164bc.

Signed-off-by: Adam Dabrowski <adam.dabrowski@robotec.ai>

* Revert "Added documentation covering ROS2 gem. (#2191)"

This reverts commit 397fca0.

Signed-off-by: Adam Dabrowski <adam.dabrowski@robotec.ai>

---------

Signed-off-by: Adam Dabrowski <adam.dabrowski@robotec.ai>
Fixing syntac error in to do shortcode

Signed-off-by: GPC-94301 <99716227+LB-JaroslawGaweda@users.noreply.github.com>
@FiniteStateGit
FiniteStateGit requested review from a team as code owners February 24, 2023 18:29
LB-JaroslawGaweda and others added 2 commits February 24, 2023 10:31
…2274)

Fixing syntax errors

Signed-off-by: GPC-94301 <99716227+LB-JaroslawGaweda@users.noreply.github.com>
* Revert "Merge branch 'development' into main"

This reverts commit a2cad04, reversing
changes made to 54e882e.

Signed-off-by: chanmosq <75444793+chanmosq@users.noreply.github.com>

* Apply suggestions from code review

Co-authored-by: Willow Hayward <17654918+willihay@users.noreply.github.com>
Signed-off-by: chanmosq <75444793+chanmosq@users.noreply.github.com>

---------

Signed-off-by: chanmosq <75444793+chanmosq@users.noreply.github.com>
Co-authored-by: Willow Hayward <17654918+willihay@users.noreply.github.com>
@willihay
willihay requested review from a team as code owners February 24, 2023 22:07
willihay and others added 2 commits February 24, 2023 14:47
* Update CODEOWNERS to fix auto reviewer assignment issues

Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>

* Update .github/CODEOWNERS

Co-authored-by: chanmosq <75444793+chanmosq@users.noreply.github.com>
Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>

---------

Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>
Co-authored-by: chanmosq <75444793+chanmosq@users.noreply.github.com>
* Added actor asset processing tutorial, as well as some updates to physx and mesh processing for continuity.

Signed-off-by: Mike Cronin <58789750+micronAMZN@users.noreply.github.com>

* Integrtating review feedback. Added an entry to the top level tutorials index which was missed previously.

Signed-off-by: Mike Cronin <58789750+micronAMZN@users.noreply.github.com>

* integrated review feedback

Signed-off-by: Mike Cronin <58789750+micronAMZN@users.noreply.github.com>

---------

Signed-off-by: Mike Cronin <58789750+micronAMZN@users.noreply.github.com>

@lmbr-pip lmbr-pip left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Most of the changes in the AWS docs look wrong.

We are deleting valid content, or reverting changes that have previously been made.

* {{< icon "processing.svg" >}} **Processing** - The job is in progress.
* {{< icon "warning-yellow.svg" >}} **Completed - Warning** - The job has completed but has emitted a warning.
* {{< icon "error.svg" >}} **Completed - Error** - The job has completed and has emitted an error.
* {{< icon "warning-yellow.svg" >}} **Completed - Error** - The job has completed and has emitted an error.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Are these the right icons? We did update to the correct icons in one of these docs.

* Source assets are files that exist in scan directories for an O3DE project, that match an Asset Builder's registered pattern, and result in a job being created when Create Jobs is called with the source asset for the Asset Builder.
* Read more about [source assets here](/docs/user-guide/assets/pipeline/source-assets/).
* Source assets are files that exist in scan directories for an O3DE project, that match an Asset Builder's registered pattern, and result in a job being created when Create Jobs is called with the source asset for the Asset Builder.
* Read more about [source assets here.](/docs/user-guide/assets/pipeline/source-assets/)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: Looks like a mistake to have the fullstop inside the link

* Intermediate assets are source assets that are generated as a product of an asset processing job.
* Read more about [intermediate assets here](/docs/user-guide/assets/pipeline/intermediate-assets/).
* Intermediate assets are source assets that are generated as a product of an asset processing job.
* Read more about [intermediate assets here.](/docs/user-guide/assets/pipeline/intermediate-assets/)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: same issue with full stop

LODs are optimized meshes with progressively lower polygon counts, fewer and smaller textures, and simplified materials. The farther an entity is from the camera, the less detail is required from the meshes that make up the entity. As the entity moves farther from the camera, it swaps to progressively simpler LOD.

With [Scene Settings](/docs/user-guide/assets/scene-settings/scene-settings/), you can specify up to five LODs (not including the base mesh) that are numbered \[`0`\] to \[`4`\], with \[`0`\] being the *highest* level of detail. LODs are not required but are recommended because they help get the best performance and visual fidelity across a range of platforms with different hardware capabilities.
You can specify up to five LODs (not including the base mesh) that are numbered \[`0`\] to \[`4`\], with \[`0`\] being the *highest* level of detail. LODs are not required but are recommended because they help get the best performance and visual fidelity across a range of platforms with different hardware capabilities.

@lmbr-pip lmbr-pip Feb 27, 2023

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: Missing link? I assume link is something we want to preserve?

These instructions use the following example CMake cache values:

* `O3DE_INSTALL_ENGINE_NAME`: "MyO3DE"
* `LY_VERSION_ENGINE_NAME`: "MyO3DE"

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Wrong name, should be O3DE version


* Choose the {{< icon browse-edit-select-files.svg >}} **Selection list** button to specify the meshes to include in the LOD.

You can also define your optimized mesh names using a _soft naming convention_, which defines naming rules for automatically adding a Level of Detail modifier and assigning the meshes to the appropriate LOD slots. LODs are ordered from 0 (the *highest* level of detail), followed by progressively *lower* levels of detail to meet the O3DE application needs.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We don't want to delete this. This is valid information.


If you are using Google, you must also use the app secret you were given for your account as the value for `ClientSecret`.

When deploying the optional AWS Cloud Development Kit (AWS CDK) application, be sure to use the AWS CDK constant that corresponds to your selected provider. Refer to the AWS CDK application deployment step in [Setting Up Client Auth](./setup/) for details.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

All the changes here should be preserved. Should be AWS CDK etc.

toc: true
---

The [AWS Cloud Development Kit](https://docs.aws.amazon.com/cdk/v2/guide/home.html) (AWS CDK) is a software development framework from AWS for defining cloud infrastructure for your project and provisioning it through **AWS CloudFormation**.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

All these changes should be preserved

| [Configuring a Project](configuration) | How to add and enable the O3DE Multiplayer Gem in a project. |
| [Running Multiplayer Projects](running) | How to run projects that use the O3DE Multiplayer Gem. |
| [Multiplayer Auto-components](autocomponents) | How to automatically create components for use with the Multiplayer Gem using the AzAutoGen system. |
| [Separating Client and Server](code_separation) | How to separate client and server logic and build dependencies.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This should be retained, want client/server split docs to part of published docs.

@@ -1,225 +0,0 @@
---

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This file should be maintained


Using these features, you can configure the Editor to connect to a compatible remote host if you wanted or a specific server executable. For example, you could configure your profile Editor to connect to a debug server.

![The Editor connected to a dedicated server](/images/user-guide/networking/multiplayer/editor_client_with_dedicated_server_mode.png)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is valid content and should not be deleted.

| sv_isDedicated | Whether the host command creates an independent or client hosted server. | True | |
| sv_isTransient | Whether a dedicated server shuts down if all existing connections disconnect. | True ||
| sv_serverSendRateMs | Minimum number of milliseconds between each network update. | 50 ms ||
| sv_versionMismatch_autoDisconnect | Determines if a mismatched connection will automatically terminate. It's recommended to keep this true; even minor differences between the version of a multiplayer component can cause unexpected behavior. | True ||

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is a valid cvar

@tonybalandiuk

Copy link
Copy Markdown

@chanmosq @micronAMZN do you know the background on why this PR was generated and why main had changes not in the development branch?

@chanmosq

chanmosq commented Feb 28, 2023

Copy link
Copy Markdown
Contributor

@chanmosq @micronAMZN do you know the background on why this PR was generated and why main had changes not in the development branch?

This PR is generated automatically and syncs main back into development (See #2245)
It's very common for main and development to diverge because docs PRs regarding existing features go into main, while docs PRs regarding code in development go into development.

Any updates you see in this PR are a result of other PRs. This one simply merges them into development.

Overall, the intent is to keep development synced with main. If anyone has suggestions on how to do this another way, please let me know!

@tonybalandiuk

Copy link
Copy Markdown

docs PRs regarding existing features go into main,

@chanmosq Following Gitflow, you could treat these docs regarding existing features as "hotfixes", in which case they would be applied to both main and dev. There's overhead in this, of course but it could be easier to manage than these larger merges. See the "Hotfix branches" in this link. https://www.atlassian.com/git/tutorials/comparing-workflows/gitflow-workflow

@willihay willihay added the do-not-merge/in-review Do not merge this PR. It is still being reviewed. label Feb 28, 2023
Comment on lines -48 to -52
### Creating a new image

To create a new image, press the **Create New Image...** button. You will be prompted for an image width and height in pixels, then for a location to save the image. If the image is saved into a [source asset directory](/docs/user-guide/assets/pipeline/scan-directories/) that is used by the project, the Terrain Macro Material will automatically populate the **Color Texture** field with the saved image.

## Editing an image

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

None of this should be reverted, it's all valid content.

| `ed_paintBrushManipulatorInnerColor` | The color of the inner circle of the Paint Brush, which is used to show where the hardness falloff begins. |
| `ed_paintBrushManipulatorOuterColor` | The color of the outer circle of the Paint Brush, which shows the overall Paint Brush size and where the hardness falloff ends. |

## PaintBrushSessionBus

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This shouldn't be reverted

| `Bicubic` | The `Bicubic` filter smooths out the image by requesting sixteen points in a grid around a requested pixel and then performing Catmull-Rom interpolation between the points.<br><br>Because this uses non-linear interpolation, there are no noticeable plus-shaped artifacts in the smoothed data. `Bicubic` sampling is the best choice when quality is needed over performance. | 16x | Great | ![Image Gradient using bicubic interpolation](/images/user-guide/components/reference/gradients/image-gradient-component-bicubic.png) |

### Creating a new image

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This shouldn't be reverted

| **Preview Position** | Sets the world location of the preview.<br> <br>*This field is available only if there is no entity selected in **Pin Preview to Shape**.* | Vector3: -Infinity to Infinity | X:`0.0`, Y:`0.0`, Z:`0.0` |
| **Preview Size** | Sets the dimensions of the preview.<br> <br>*This field is available only if there is no entity selected in **Pin Preview to Shape**.* | Vector3: 0.0 to Infinity | X:`1.0`, Y:`1.0`, Z:`1.0` |
| **Constrain to Shape** | If `Enabled`, the gradient preview uses the bounds of the entity selected in **Pin Preview to Shape**.<br> <br>*This field is available only if an entity is selected in **Pin Preview to Shape**.* | Boolean | `Disabled` |
| **Source Type** | Determines whether to create a new image or use an existing image. | `Create New Image`, `Use Existing Image` | `Use Existing Image` |

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This shouldn't be reverted

chanmosq and others added 2 commits February 28, 2023 11:55
Signed-off-by: chanmosq <75444793+chanmosq@users.noreply.github.com>
* Fix broken link in Build section

Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>

* Fix broken and outdated links

Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>

---------

Signed-off-by: Willow Hayward <17654918+willihay@users.noreply.github.com>
@chanmosq

Copy link
Copy Markdown
Contributor

@lmbr-pip @tonybalandiuk @mbalfour-amzn Sorry for the mishap here - a commit was pulled in that shouldn't have been. Closing this PR.

@chanmosq chanmosq closed this Feb 28, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/in-review Do not merge this PR. It is still being reviewed.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

9 participants