Skip to content

Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to 2c2fa53 - #570

Open
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
release-1.2from
konflux/mintmaker/release-1.2/registry.access.redhat.com-ubi8-ubi-minimal-latest
Open

Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to 2c2fa53#570
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
release-1.2from
konflux/mintmaker/release-1.2/registry.access.redhat.com-ubi8-ubi-minimal-latest

Conversation

@red-hat-konflux-kflux-prd-rh03

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot commented Sep 1, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi8/ubi-minimal final digest c2483a82c2fa53

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "after 5am on tuesday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@openshift-ci
openshift-ci Bot requested review from Thealisyed and gcs278 September 1, 2026 08:07
@openshift-ci openshift-ci Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Sep 1, 2026
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown
Contributor

Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR.

I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/mintmaker/release-1.2/registry.access.redhat.com-ubi8-ubi-minimal-latest branch from 22a7b41 to 627d22c Compare September 1, 2026 12:02
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to aacf6ef Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to 128cb37 Sep 1, 2026
@gcs278

gcs278 commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

/assign @grzpiotrowski

@gcs278

gcs278 commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

/assign @pedjak

@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/mintmaker/release-1.2/registry.access.redhat.com-ubi8-ubi-minimal-latest branch from 627d22c to 0f52a4a Compare September 8, 2026 08:02
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to 128cb37 Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to d9beb74 Sep 8, 2026
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please ask for approval from grzpiotrowski. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

…st to 2c2fa53

Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/mintmaker/release-1.2/registry.access.redhat.com-ubi8-ubi-minimal-latest branch from 0f52a4a to b63fae1 Compare September 8, 2026 20:02
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to d9beb74 Update registry.access.redhat.com/ubi8/ubi-minimal:latest Docker digest to 2c2fa53 Sep 8, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants