Skip to content

Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 2404ca0 - #574

Open
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
mainfrom
konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-3-rhel-9-component-update-external-dns-container-ext-dns-optr-1-3-rhel-9
Open

Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 2404ca0#574
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
mainfrom
konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-3-rhel-9-component-update-external-dns-container-ext-dns-optr-1-3-rhel-9

Conversation

@red-hat-konflux-kflux-prd-rh03

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot commented Sep 1, 2026

Copy link
Copy Markdown
Contributor

Image created from 'https://github.com/openshift/external-dns?rev=d367144812a69f4ba1e6401f347044094400a6b9'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/external-dns-operator-tenant/ext-dns-optr-1-3-rhel-9/external-dns-container-ext-dns-optr-1-3-rhel-9 digest 9e02d30 -> 2404ca0

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

@openshift-ci openshift-ci Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Sep 1, 2026
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign alebedev87 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown
Contributor

Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR.

I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 12f0626 Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 4cab164 Sep 2, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-3-rhel-9-component-update-external-dns-container-ext-dns-optr-1-3-rhel-9 branch from 734a795 to fe99125 Compare September 2, 2026 08:14
Image created from 'https://github.com/openshift/external-dns?rev=d367144812a69f4ba1e6401f347044094400a6b9'

Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh03 red-hat-konflux-kflux-prd-rh03 Bot changed the title Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 4cab164 Update external-dns-container-ext-dns-optr-1-3-rhel-9 to 2404ca0 Sep 2, 2026
@red-hat-konflux-kflux-prd-rh03
red-hat-konflux-kflux-prd-rh03 Bot force-pushed the konflux/component-updates/external-dns-operator-bundle-container-ext-dns-optr-1-3-rhel-9-component-update-external-dns-container-ext-dns-optr-1-3-rhel-9 branch from fe99125 to 3c73237 Compare September 2, 2026 08:34
@gcs278

gcs278 commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

/assign @grzpiotrowski

@gcs278

gcs278 commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

/assign @pedjak

@openshift-ci openshift-ci Bot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Sep 2, 2026
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

1 similar comment
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

konflux-nudge needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants