Skip to content

NO-JIRA: [bot] Bump openshift/prometheus to v3.14.0 - #363

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
rhobs:automated-updates-main
Sep 3, 2026
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
rhobs:automated-updates-main

Conversation

@openshift-monitoring-bot

@openshift-monitoring-bot openshift-monitoring-bot Bot commented Sep 3, 2026

Copy link
Copy Markdown

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.

@openshift-merge-bot

Copy link
Copy Markdown

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci
openshift-ci Bot requested review from jan--f and marioferh September 3, 2026 00:15
@simonpasquier

Copy link
Copy Markdown

/approve
/lgtm
/verified by CI
/retitle [bot] Bump openshift/prometheus to v3.14.0

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 3, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@simonpasquier: This PR has been marked as verified by CI.

Details

In response to this:

/approve
/lgtm
/verified by CI
/retitle [bot] Bump openshift/prometheus to v3.14.0

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 3, 2026
@openshift-merge-bot

Copy link
Copy Markdown

Scheduling required tests:
/test e2e-agnostic-cmo
/test e2e-aws-ovn
/test e2e-aws-ovn-upgrade

@simonpasquier

Copy link
Copy Markdown

/test okd-scos-images

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 3, 2026
Signed-off-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
Signed-off-by: github-actions[bot] <github-actions[bot]@users.noreply.github.com>
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 3, 2026
@machine424

Copy link
Copy Markdown

changes due to rhobs/syncbot#163

/lgtm
/verified by CI
/retitle NO-JIRA: [bot] Bump openshift/prometheus to v3.14.0

@openshift-ci-robot

Copy link
Copy Markdown

@machine424: This PR has been marked as verified by CI.

Details

In response to this:

changes due to rhobs/syncbot#163

/lgtm
/verified by CI
/retitle NO-JIRA: [bot] Bump openshift/prometheus to v3.14.0

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot changed the title [bot] Bump openshift/prometheus to v3.14.0 NO-JIRA: [bot] Bump openshift/prometheus to v3.14.0 Sep 3, 2026
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 3, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

Description

This is an automated version bump from CI.
The logs for this run can be found in the syncbot repo actions.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 3, 2026
@openshift-merge-bot

Copy link
Copy Markdown

Scheduling required tests:
/test e2e-agnostic-cmo
/test e2e-aws-ovn
/test e2e-aws-ovn-upgrade

@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: machine424, openshift-monitoring-bot[bot], simonpasquier

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [machine424,simonpasquier]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

@openshift-monitoring-bot[bot]: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit c2c48fe into openshift:main Sep 3, 2026
11 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants