[s2i] Add a manila/cephfs test job - #641
openshift-merge-bot[bot] merged 1 commit into
Conversation
|
Skipping CI for Draft Pull Request. |
|
Important
This repository does not receive automatic reviews because it has fewer than 10 stars. ⚙️ Run configurationConfiguration used: Central YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Advanced Run ID: Comment |
56bd4bf to
5fd51cc
Compare
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/openstack-operator#2058 is needed. |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 38m 55s |
5fd51cc to
fc966bb
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 10m 06s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 12m 47s |
5b6bd12 to
a251dd9
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 10m 44s |
8a7d854 to
ccc9de1
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 28m 16s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 06m 16s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 20m 18s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 19m 50s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 19m 35s |
|
recheck |
ccc9de1 to
9587af7
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 01m 33s |
|
recheck |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 13m 03s |
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider FAILURE in 13m 26s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 31m 01s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 03m 19s |
b511ea2 to
b0e95f6
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 59m 03s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 58m 35s |
b0e95f6 to
e86ab31
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 36m 54s |
|
recheck |
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/openstack-operator#2058 is needed. |
|
recheck |
Temporarily skip a few tests until a newer tempest is available (currently still the antelope one). The old tempest job was renamed and kept to cross-test the old containers with the new operators. It also got a few tempest.conf overrides, especially the new endpoint needed to make it work (pending fixes on python-tempestconf). The new job is voting here on the operator for now to have a baseline. Whether additional (nfsganesha) jobs will be voting on the operator too, or whether the job will need to be tuned when running on the operator pipeline vs the component one, it's up to future evaluation.
e86ab31 to
50571e8
Compare
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 28m 09s |
|
The dependency was merged, this is ready to go (and needed to unbreak the jobs of this operator) |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: fmount, tosky The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
61de2d6
into
openstack-k8s-operators:main
Temporarily skip a few tests until a newer tempest is available
(currently still the antelope one).
The old tempest job was renamed and kept to cross-test the old
containers with the new operators. It also got a few tempest.conf
overrides, especially the new endpoint needed to make it work
(pending fixes on python-tempestconf).
The new job is voting here on the operator for now to have a baseline.
Whether additional (nfsganesha) jobs will be voting on the operator too,
or whether the job will need to be tuned when running on the operator
pipeline vs the component one, it's up to future evaluation.
Depends-On: openstack-k8s-operators/openstack-operator#2058