Conversation
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (1)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Central YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: fmount The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
we might require to include |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 03m 26s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 18m 19s |
|
Hey @fmount yeah, since the manila-share RPM doesn't contain the "manila.tests" module, the dummy driver isn't included. It'd have been perfect for this use-case. Maybe it still can be if we can shove it into the testing here, and use it via the extraMounts logic when we don't use the s2i container images? You're right that with s2i, we kinda get this package because we're closer to manila's source there - and we include everything in the package, including tests today. An LLM was tortured to get something along these lines as a brittle dev/CI-only flake we can incorporate: then: |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 57m 14s |
Thanks @gouthampacha for checking it. I think that 18 jobs can continue to use ceph backend as the goal is to remove it from main, where we will actually rely on s2i generated content. I confirmed w/ the last run that |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 22m 19s |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 03m 39s |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 03m 05s |
| - tests?\/functional | ||
| - ^renovate.json$ | ||
| vars: | ||
| cifmw_crc_additional_insecure_registries: |
There was a problem hiding this comment.
This part can be moved to a potential centralized base job that lives in s2i-openstack-containers, but I'm considering moving forward with this and have a dedicated follow up for this (which I think is already in our plans). cc @tosky
Note that we need a follow up anyway to target the ceph removal via install_yamls, but at that point this job will still be green because this patch migrates it to s2i+dummy backend
|
I'm going to put this patch on hold as we need to rebase it on top of #647 and rename the jobs. |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 28m 59s |
Signed-off-by: Francesco Pantano <fpantano@redhat.com>
Signed-off-by: Francesco Pantano <fpantano@redhat.com>
Add cifmw overrides to consume s2i images coming from content provider and have a command to patch the csv where RELATED_IMAGES_* are replaced with what s2i provides. Signed-off-by: Francesco Pantano <fpantano@redhat.com>
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 59m 57s |
Update kuttl jobs to consume s2i content