Conversation
It will make it easier to reuse them elsewhere, for example inside s2i-openstack-containers.
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: tosky The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important
This repository does not receive automatic reviews because it has fewer than 10 stars. ⚙️ Run configurationConfiguration used: Central YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Advanced Run ID: Comment |
|
Build succeeded (check pipeline). ✔️ manila-s2i-content-provider SUCCESS in 3h 28m 35s |
|
/retest-required |
|
/test golint |
|
/retest-required |
It will make it easier to reuse them elsewhere, for example inside s2i-openstack-containers.