Skip to content

feat: add on config options - #104

Merged
truongnhatquang21 merged 5 commits into
mainfrom
feat/onramp-config
Nov 14, 2025
Merged

truongnhatquang21 merged 5 commits into
mainfrom
feat/onramp-config

Conversation

@truongnhatquang21

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercel Bot commented Nov 14, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

3 Skipped Deployments
Project Deployment Preview Comments Updated (UTC)
daily-check-in-example Ignored Ignored Preview Nov 14, 2025 7:50am
ex-telegram-check-in Ignored Ignored Preview Nov 14, 2025 7:50am
waypoint-playground Ignored Ignored Preview Nov 14, 2025 7:50am

@truongnhatquang21
truongnhatquang21 merged commit 9bd984f into main Nov 14, 2025
5 checks passed
@truongnhatquang21
truongnhatquang21 deleted the feat/onramp-config branch November 14, 2025 07:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants