Skip to content

Update dependency react-router to ^7.18.3 (main) - #6809

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
konflux/mintmaker/main-main/react-router-monorepo
Sep 8, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
konflux/mintmaker/main-main/react-router-monorepo

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
react-router (source) ^7.18.2^7.18.3 age confidence

Warning

Some dependencies could not be looked up. Check the warning logs for more information.


Release Notes

remix-run/react-router (react-router)

v7.18.3: v7.18.3

Compare Source

See the changelog for release notes: https://github.com/remix-run/react-router/blob/v7/CHANGELOG.md#v7183


Configuration

📅 Schedule: (in timezone America/New_York)

  • Branch creation
    • "every weekend"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about these updates again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/main-main/react-router-monorepo branch from 96d9538 to 12cd6ed Compare September 5, 2026 06:15
@sonarqubecloud

sonarqubecloud Bot commented Sep 5, 2026

Copy link
Copy Markdown

@KevinFCormier

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: KevinFCormier, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Sep 8, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 09c027b into main Sep 8, 2026
21 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the konflux/mintmaker/main-main/react-router-monorepo branch September 8, 2026 13:02
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant