Skip to content

Update module github.com/bytedance/sonic/loader to v0.5.2 - #2336

Merged
openshift-merge-bot[bot] merged 1 commit into
backplane-5.0from
konflux/mintmaker/backplane-5.0/github.com-bytedance-sonic-loader-0.x
Aug 27, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
backplane-5.0from
konflux/mintmaker/backplane-5.0/github.com-bytedance-sonic-loader-0.x

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
github.com/bytedance/sonic/loader v0.5.1v0.5.2 age confidence

Configuration

📅 Schedule: (in timezone America/New_York)

  • Branch creation
    • "before 8am on tuesday and thursday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@dislbenn

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented Aug 27, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dislbenn, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved PR approval has been given label Aug 27, 2026
@sonarqubecloud

Copy link
Copy Markdown

@openshift-merge-bot
openshift-merge-bot Bot merged commit 41ed296 into backplane-5.0 Aug 27, 2026
14 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the konflux/mintmaker/backplane-5.0/github.com-bytedance-sonic-loader-0.x branch August 27, 2026 17:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant