Skip to content

Update github.com/stolostron/klusterlet-addon-controller digest to 5341d76 - #2342

Merged
openshift-merge-bot[bot] merged 1 commit into
backplane-5.0from
konflux/mintmaker/backplane-5.0/github.com-stolostron-klusterlet-addon-controller-digest
Sep 2, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
backplane-5.0from
konflux/mintmaker/backplane-5.0/github.com-stolostron-klusterlet-addon-controller-digest

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
github.com/stolostron/klusterlet-addon-controller require digest 83876f15341d76

Configuration

📅 Schedule: (in timezone America/New_York)

  • Branch creation
    • "before 8am on tuesday and thursday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

…41d76

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@sonarqubecloud

sonarqubecloud Bot commented Sep 1, 2026

Copy link
Copy Markdown

@dislbenn

dislbenn commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dislbenn, red-hat-konflux[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved PR approval has been given label Sep 2, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit e9e8757 into backplane-5.0 Sep 2, 2026
14 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the konflux/mintmaker/backplane-5.0/github.com-stolostron-klusterlet-addon-controller-digest branch September 2, 2026 16:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant