Skip to content

fix: merge CODEOWNERS - #241

Merged
gabrielcocenza merged 1 commit into
mainfrom
fix/codeowners
Jul 21, 2026
Merged

gabrielcocenza merged 1 commit into
mainfrom
fix/codeowners

Conversation

@dariofaccin

Copy link
Copy Markdown
Contributor

This PR fixes the CODEOWNERS file so that MLOps and SolEng are set as reviewers for the PRs.

@ckfbot

ckfbot commented Jul 21, 2026

Copy link
Copy Markdown

🤖 Backport labels populated

Labels to this pull request were added automatically by the populate-labels.yaml action.

When the PR is merged, backport PRs according to the labels will be automatically created. To skip the backport creation, remove any unneeded labels before merging the PR.

@gabrielcocenza
gabrielcocenza merged commit 79791ee into main Jul 21, 2026
17 checks passed
@gabrielcocenza
gabrielcocenza deleted the fix/codeowners branch July 21, 2026 12:11
@ckfbot

ckfbot commented Jul 21, 2026

Copy link
Copy Markdown

Successfully created backport PR for track/1.15:

dariofaccin added a commit that referenced this pull request Jul 21, 2026
# Description
Backport of #241 to `track/1.15`.
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants