Skip to content

fix: merge CODEOWNERS (backport #241) - #242

Merged
dariofaccin merged 1 commit into
track/1.15from
backport-241-to-track/1.15
Jul 21, 2026
Merged

dariofaccin merged 1 commit into
track/1.15from
backport-241-to-track/1.15

Conversation

@ckfbot

@ckfbot ckfbot commented Jul 21, 2026

Copy link
Copy Markdown

Description

Backport of #241 to track/1.15.

(cherry picked from commit 79791ee)
@ckfbot
ckfbot requested a review from a team as a code owner July 21, 2026 12:18
@ckfbot
ckfbot requested a review from dariofaccin July 21, 2026 12:18
@ckfbot ckfbot mentioned this pull request Jul 21, 2026
@dariofaccin
dariofaccin merged commit d1a0a90 into track/1.15 Jul 21, 2026
32 of 33 checks passed
@dariofaccin
dariofaccin deleted the backport-241-to-track/1.15 branch July 21, 2026 14:05
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants